ticdc: update ticdc-debezium.md - #19544
Merged
Merged
Conversation
wk989898
marked this pull request as draft
December 2, 2024 04:29
wk989898
marked this pull request as ready for review
December 9, 2024 07:19
Member
|
Note: Do not use hard Tabs in our documentation. Hard Tabs make the CI check fail. I've replaced each hard Tab with four spaces via the 852aa42 commit. |
hfxsd
reviewed
Feb 7, 2025
hfxsd
reviewed
Feb 7, 2025
Co-authored-by: xixirangrang <hfxsd@hotmail.com>
lilin90
reviewed
Feb 7, 2025
lilin90
reviewed
Feb 7, 2025
|
|
||
| - Debezium converts `charsetName` to `"utf8mb4"` when the column collation is `"utf8_unicode_ci"` and the character set is null, but TiCDC does not. | ||
|
|
||
| - Debezium escapes ENUM elements, but TiCDC does not. For example, Debezium encodes ENUM elements ('c', 'd', 'g,''h') to ('c','d','g,\'\'h'). |
Member
There was a problem hiding this comment.
Please update the example @wk989898. It does not display well.

hfxsd
approved these changes
Feb 7, 2025
wk989898
commented
Feb 7, 2025
Co-authored-by: nhsmw <nhsmwk@gmail.com>
lilin90
reviewed
Feb 7, 2025
wk989898
commented
Feb 7, 2025
Co-authored-by: nhsmw <nhsmwk@gmail.com>
Member
|
Content is ready to merge except the v9.0.0 version number. |
wk989898
commented
Feb 28, 2025
Co-authored-by: nhsmw <nhsmwk@gmail.com>
lilin90
reviewed
Mar 12, 2025
Member
|
/unhold |
Member
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: lilin90 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This was referenced Aug 13, 2026
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
First-time contributors' checklist
What is changed, added or deleted? (Required)
zh: pingcap/docs-cn#19704
Which TiDB version(s) do your changes apply to? (Required)
Tips for choosing the affected version(s):
By default, CHOOSE MASTER ONLY so your changes will be applied to the next TiDB major or minor releases. If your PR involves a product feature behavior change or a compatibility change, CHOOSE THE AFFECTED RELEASE BRANCH(ES) AND MASTER.
For details, see tips for choosing the affected versions.
What is the related PR or file link(s)?
Do your changes match any of the following descriptions?